Skip to content

feat: Migrate cloud credentials to the new collection - #32122

Closed
Gustrb wants to merge 26 commits into
developfrom
feature/migrate-cloud-credentials
Closed

Gustrb wants to merge 26 commits into
developfrom
feature/migrate-cloud-credentials

Conversation

@Gustrb

@Gustrb Gustrb commented Apr 3, 2024

Copy link
Copy Markdown
Contributor

This PR is a children of: #31986 so it must be merged before this one gets merged.

Proposed changes (including videos or screenshots)

Since all the cloud credentials data must be stored inside the rocketchat_workspace_credentials collection, we must migrate all the values from inside the settings collection to their new home.

Issue(s)

Steps to test or reproduce

Further comments

CONN-5

@dionisio-bot

dionisio-bot Bot commented Apr 3, 2024 •

Copy link
Copy Markdown
Contributor

Looks like this PR is not ready to merge, because of the following issues:

  • This PR is missing the 'stat: QA assured' label
  • This PR is missing the required milestone or project

Please fix the issues and try again

If you have any trouble, please check the PR guidelines

@changeset-bot

changeset-bot Bot commented Apr 3, 2024 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 865a9d4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 32 packages
Name Type
@rocket.chat/meteor Minor
@rocket.chat/core-typings Minor
@rocket.chat/model-typings Minor
@rocket.chat/models Minor
@rocket.chat/apps Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Major
@rocket.chat/livechat Patch
@rocket.chat/rest-typings Minor
@rocket.chat/ui-contexts Major
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/instance-status Patch
@rocket.chat/fuselage-ui-kit Major
@rocket.chat/ddp-client Patch
@rocket.chat/ui-avatar Major
@rocket.chat/ui-client Major
@rocket.chat/ui-video-conf Major
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Major

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@Gustrb
Gustrb requested a review from ggazzo April 3, 2024 19:42
@codecov

codecov Bot commented Apr 3, 2024 •

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 55.65%. Comparing base (ff4e396) to head (865a9d4).
Report is 47 commits behind head on develop.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #32122      +/-   ##
===========================================
+ Coverage    55.59%   55.65%   +0.05%     
===========================================
  Files         2407     2429      +22     
  Lines        52905    53396     +491     
  Branches     10866    10983     +117     
===========================================
+ Hits         29415    29717     +302     
- Misses       20886    21045     +159     
- Partials      2604     2634      +30     
Flag Coverage Δ
e2e 55.24% <ø> (+0.36%) ⬆️
unit 72.73% <ø> (-0.84%) ⬇️

Flags with carried forward coverage won't be shown. Click here to find out more.

@casalsgh casalsgh modified the milestones: 7.0, 6.8 Apr 5, 2024
@Gustrb
Gustrb marked this pull request as ready for review April 5, 2024 12:46
@Gustrb
Gustrb requested review from a team as code owners April 5, 2024 12:46
@casalsgh casalsgh removed this from the 6.8 milestone Apr 22, 2024
@debdutdeb
debdutdeb self-requested a review April 22, 2024 20:28
Comment thread .changeset/fifty-laws-sparkle.md Outdated

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I believe we're trying to only add migrations in the next major release 😬
Is there any chance we could move this to the release-7.0.0 branch?
cc @sampaiodiego

Co-authored-by: Matheus Barbosa Silva <36537004+matheusbsilva137@users.noreply.github.com>
@Gustrb Gustrb closed this Jul 22, 2024
@tassoevan
tassoevan deleted the feature/migrate-cloud-credentials branch July 30, 2025 15:10
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants