fix: Region Select change that is causing test failures - #12622
Merged
bnussman-akamai merged 1 commit intoJul 31, 2025
Merged
Conversation
bnussman-akamai
requested review from
cpathipa and
mjac0bs
and removed request for
a team
July 31, 2025 21:39
jdamore-linode
approved these changes
Jul 31, 2025
jdamore-linode
left a comment
Contributor
There was a problem hiding this comment.
Thanks for the quick fix @bnussman-akamai! I see some of the impacted tests passing now 🚀
mjac0bs
approved these changes
Jul 31, 2025
mjac0bs
left a comment
Contributor
There was a problem hiding this comment.
Approving assuming CI finishes doing its thing and continues to look good - should since tests are over halfway through and passing.
Do you know what about that change caused tests to break?
Cloud Manager UI test results🔺 1 failing test on test run #1 ↗︎
Details
TroubleshootingUse this command to re-run the failing tests: pnpm cy:run -s "cypress/e2e/core/linodes/linode-storage.spec.ts" |
|||||||||||||||||
Member
Author
|
@mjac0bs From Joe:
|
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Description 📝
autoHighlightin fix: [M3-10264] - Disabled Autocomplete Options #12583 and I believe it broke a ton of cypress tests (thanks for catching @jdamore-linode )How to test 🧪