Skip to content

Auto-generate registry credentials instead of hardcoding - #1956

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-metal3:masterfrom
dtantsur:registry-pass
Sep 4, 2026
Merged

Auto-generate registry credentials instead of hardcoding#1956
openshift-merge-bot[bot] merged 1 commit into
openshift-metal3:masterfrom
dtantsur:registry-pass

Conversation

@dtantsur

@dtantsur dtantsur commented Sep 2, 2026

Copy link
Copy Markdown
Member

Signed-off-by: Dmitry Tantsur dtantsur@protonmail.com

Signed-off-by: Dmitry Tantsur <dtantsur@protonmail.com>
@dtantsur

dtantsur commented Sep 3, 2026

Copy link
Copy Markdown
Member Author

/retest

@MahnoorAsghar

Copy link
Copy Markdown
Contributor

/lgtm
I dont think the 2 failures are related..otherwise many more jobs would fail I think

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 3, 2026
@dtantsur

dtantsur commented Sep 3, 2026

Copy link
Copy Markdown
Member Author

/cc @andfasano @bfournie

PTAL for approval, this change may have broader impact

@openshift-ci
openshift-ci Bot requested a review from bfournie September 3, 2026 14:19
@bfournie

bfournie commented Sep 3, 2026

Copy link
Copy Markdown
Contributor

/approve

@openshift-ci

openshift-ci Bot commented Sep 3, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: bfournie

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 3, 2026
@bfournie

bfournie commented Sep 4, 2026

Copy link
Copy Markdown
Contributor

/retest-required

@openshift-ci

openshift-ci Bot commented Sep 4, 2026

Copy link
Copy Markdown

@dtantsur: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-agent-5control-ipv4 bc8f189 link false /test e2e-agent-5control-ipv4
ci/prow/e2e-metal-ipi-ovn-dualstack bc8f189 link false /test e2e-metal-ipi-ovn-dualstack

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@dtantsur

dtantsur commented Sep 4, 2026

Copy link
Copy Markdown
Member Author

/test e2e-metal-ipi-ovn-ipv4

Go mirrors still

@openshift-merge-bot
openshift-merge-bot Bot merged commit 25732d2 into openshift-metal3:master Sep 4, 2026
17 of 19 checks passed
@dtantsur
dtantsur deleted the registry-pass branch September 7, 2026 11:27
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants