Skip to content

Automated - Update config ./config/openshift.yaml - #3963

Merged
openshift-merge-bot[bot] merged 1 commit into
mainfrom
automated---update-config-./config/openshift.yaml
Aug 31, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
mainfrom
automated---update-config-./config/openshift.yaml

Conversation

@openshift-ci

@openshift-ci openshift-ci Bot commented Aug 27, 2026

Copy link
Copy Markdown
Contributor

No description provided.

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Pipeline controller notification
This repo is configured to use the pipeline controller. Second-stage tests will be triggered either automatically or after lgtm label is added, depending on the repository configuration. The pipeline controller will automatically detect which contexts are required and will utilize /test Prow commands to trigger the second stage.

For optional jobs, comment /test ? to see a list of all defined jobs. To trigger manually all jobs from second stage use /pipeline required command.

This repository is configured in: automatic mode

@openshift-ci openshift-ci Bot added the ready-for-human-review Indicates a PR has been reviewed by automated tools and is ready for human review label Aug 27, 2026
@openshift-ci
openshift-ci Bot requested review from deads2k and sosiouxme August 27, 2026 22:10
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Scheduling required tests:
/test e2e

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD 4ba9539 and 2 for PR HEAD 6c1b4c1 in total

@openshift-ci
openshift-ci Bot force-pushed the automated---update-config-./config/openshift.yaml branch from 6c1b4c1 to cd6f998 Compare August 28, 2026 10:09
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Scheduling required tests:
/test e2e

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD 4ba9539 and 2 for PR HEAD cd6f998 in total

@openshift-ci
openshift-ci Bot force-pushed the automated---update-config-./config/openshift.yaml branch from cd6f998 to 13cfea5 Compare August 28, 2026 22:05
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Scheduling required tests:
/test e2e

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD 4ba9539 and 2 for PR HEAD 13cfea5 in total

@openshift-ci
openshift-ci Bot force-pushed the automated---update-config-./config/openshift.yaml branch from 13cfea5 to 6269c8b Compare August 29, 2026 10:05
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Scheduling required tests:
/test e2e

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD 4ba9539 and 2 for PR HEAD 6269c8b in total

@openshift-ci
openshift-ci Bot force-pushed the automated---update-config-./config/openshift.yaml branch from 6269c8b to 051f674 Compare August 29, 2026 22:06
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Scheduling required tests:
/test e2e

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD 4ba9539 and 2 for PR HEAD 051f674 in total

@openshift-ci
openshift-ci Bot force-pushed the automated---update-config-./config/openshift.yaml branch from 051f674 to b933813 Compare August 30, 2026 10:11
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Scheduling required tests:
/test e2e

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD 4ba9539 and 2 for PR HEAD b933813 in total

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD 1acf8f9 and 1 for PR HEAD b933813 in total

@openshift-ci
openshift-ci Bot force-pushed the automated---update-config-./config/openshift.yaml branch from b933813 to 3a971e4 Compare August 30, 2026 22:06
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Scheduling required tests:
/test e2e

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD 1acf8f9 and 2 for PR HEAD 3a971e4 in total

@openshift-ci
openshift-ci Bot force-pushed the automated---update-config-./config/openshift.yaml branch from 3a971e4 to 38dd0e3 Compare August 31, 2026 10:08
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Scheduling required tests:
/test e2e

@neisw

neisw commented Aug 31, 2026

Copy link
Copy Markdown
Contributor

/override ci/prow/e2e

@neisw

neisw commented Aug 31, 2026

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci

openshift-ci Bot commented Aug 31, 2026

Copy link
Copy Markdown
Contributor Author

@neisw: Overrode contexts on behalf of neisw: ci/prow/e2e

Details

In response to this:

/override ci/prow/e2e

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 31, 2026
@openshift-ci

openshift-ci Bot commented Aug 31, 2026

Copy link
Copy Markdown
Contributor Author

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: neisw, openshift-ci[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 31, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit 7ecd967 into main Aug 31, 2026
11 checks passed
@openshift-ci

openshift-ci Bot commented Aug 31, 2026

Copy link
Copy Markdown
Contributor Author

@openshift-ci[bot]: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. ready-for-human-review Indicates a PR has been reviewed by automated tools and is ready for human review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant