Skip to content

[devscripts] Parametrize devscripts's ext_iface - #3809

Merged
openshift-merge-bot[bot] merged 1 commit into
openstack-k8s-operators:mainfrom
bogdando:devscripts
Mar 30, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
openstack-k8s-operators:mainfrom
bogdando:devscripts

Conversation

@bogdando

Copy link
Copy Markdown
Contributor

Replace the hardcoded value with a variable.
This value may need changes for SNO BM cases.

Required-by: #3739

@bogdando
bogdando requested review from evallesp and rebtoor March 30, 2026 09:35
@bogdando bogdando changed the title Parametrize cifmw_devscripts_ext_iface [devscripts] Parametrize devscripts's ext_iface Mar 30, 2026
evallesp
evallesp previously approved these changes Mar 30, 2026

@evallesp evallesp left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

Replace the hardcoded value with a variable.
This value may need changes for SNO BM cases.

Signed-off-by: Bohdan Dobrelia <bdobreli@redhat.com>

@evallesp evallesp left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci openshift-ci Bot added the lgtm label Mar 30, 2026

@michburk michburk left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@openshift-ci

openshift-ci Bot commented Mar 30, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: michburk

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit 6a339e4 into openstack-k8s-operators:main Mar 30, 2026
11 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants