Skip to content

Fixes value for storage.oci.repository - #974

Merged
tekton-robot merged 1 commit into
tektoncd:mainfrom
PuneetPunamiya:fix-storage-oci-repo
Nov 22, 2023
Merged

tekton-robot merged 1 commit into
tektoncd:mainfrom
PuneetPunamiya:fix-storage-oci-repo

Conversation

@PuneetPunamiya

@PuneetPunamiya PuneetPunamiya commented Oct 27, 2023

Copy link
Copy Markdown
Member
  • Previously while providing repo url value for storage oci repository, chains controller was giving an error as
    a digest must contain exactly one '@' separator (e.g. registry/repository@digest)
    because, it was not able to add digest and repo name

  • Hence this patch fixes it, by formatting the value provided by the user and thus storing the
    attestations/signatures in the provide location

Changes

Submitter Checklist

As the author of this PR, please check off the items in this checklist:

  • Has Docs included if any changes are user facing
  • Has Tests included if any functionality added or changed
  • Follows the commit message standard
  • Meets the Tekton contributor standards (including
    functionality, content, code)
  • Release notes block below has been updated with any user facing changes (API changes, bug fixes, changes requiring upgrade notices or deprecation warnings)
  • Release notes contains the string "action required" if the change requires additional action from users switching to the new release

Release Notes

NONE

@tekton-robot tekton-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Oct 27, 2023
@tekton-robot

Copy link
Copy Markdown
Contributor

The following is the coverage report on the affected files.
Say /test pull-tekton-chains-go-coverage to re-run this coverage report

File Old Coverage New Coverage Delta
pkg/chains/storage/oci/legacy.go 37.7% 37.1% -0.7

Comment thread pkg/chains/storage/oci/legacy.go Outdated
Comment thread pkg/chains/storage/oci/legacy.go Outdated

@lcarva lcarva left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I'm not sure how this fixes the issue from the description. A unit test might be a good idea to express it.

@PuneetPunamiya

Copy link
Copy Markdown
Member Author

I'm not sure how this fixes the issue from the description. A unit test might be a good idea to express it.

Yeah will be adding it 👍🏻

@tekton-robot tekton-robot added size/M Denotes a PR that changes 30-99 lines, ignoring generated files. and removed size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Nov 9, 2023
@tekton-robot

Copy link
Copy Markdown
Contributor

The following is the coverage report on the affected files.
Say /test pull-tekton-chains-go-coverage to re-run this coverage report

File Old Coverage New Coverage Delta
pkg/chains/storage/oci/legacy.go 37.7% 40.5% 2.8

@PuneetPunamiya

Copy link
Copy Markdown
Member Author

I'm not sure how this fixes the issue from the description. A unit test might be a good idea to express it.

Yeah will be adding it 👍🏻

I have added the test cases. PTAL

@tekton-robot

Copy link
Copy Markdown
Contributor

The following is the coverage report on the affected files.
Say /test pull-tekton-chains-go-coverage to re-run this coverage report

File Old Coverage New Coverage Delta
pkg/chains/storage/oci/legacy.go 37.7% 40.0% 2.3

Comment thread pkg/chains/storage/oci/legacy.go Outdated
@tekton-robot

Copy link
Copy Markdown
Contributor

The following is the coverage report on the affected files.
Say /test pull-tekton-chains-go-coverage to re-run this coverage report

File Old Coverage New Coverage Delta
pkg/chains/storage/oci/legacy.go 37.7% 40.0% 2.3

@tekton-robot tekton-robot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. and removed size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Nov 10, 2023
@tekton-robot

Copy link
Copy Markdown
Contributor

The following is the coverage report on the affected files.
Say /test pull-tekton-chains-go-coverage to re-run this coverage report

File Old Coverage New Coverage Delta
pkg/chains/storage/oci/legacy.go 37.7% 38.7% 1.0

Comment thread pkg/chains/storage/oci/legacy.go Outdated
Comment thread pkg/chains/storage/oci/legacy.go Outdated
Comment thread pkg/chains/storage/oci/legacy_test.go Outdated
Comment thread pkg/chains/storage/oci/legacy_test.go Outdated
Comment thread pkg/chains/storage/oci/legacy_test.go Outdated
Comment thread pkg/chains/storage/oci/legacy.go
@tekton-robot

Copy link
Copy Markdown
Contributor

The following is the coverage report on the affected files.
Say /test pull-tekton-chains-go-coverage to re-run this coverage report

File Old Coverage New Coverage Delta
pkg/chains/storage/oci/legacy.go 37.7% 41.3% 3.6

Comment thread pkg/chains/storage/oci/legacy.go Outdated
Comment thread pkg/chains/storage/oci/legacy.go Outdated
Previously while providing repo url value for storage
oci repository, chains controller was giving an error
as

`a digest must contain exactly one '@' separator (e.g.
registry/repository@digest)`

because, it was not able to add digest

Hence this patch fixes it, by formatting the value
provided by the user and thus storing the
attestations/signatures in the provided location

Signed-off-by: PuneetPunamiya <ppunamiy@redhat.com>
@tekton-robot tekton-robot added size/M Denotes a PR that changes 30-99 lines, ignoring generated files. and removed size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Nov 22, 2023
@tekton-robot

Copy link
Copy Markdown
Contributor

The following is the coverage report on the affected files.
Say /test pull-tekton-chains-go-coverage to re-run this coverage report

File Old Coverage New Coverage Delta
pkg/chains/storage/oci/legacy.go 37.7% 40.0% 2.3

@lcarva lcarva left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@tekton-robot tekton-robot added the lgtm Indicates that a PR is ready to be merged. label Nov 22, 2023
@tekton-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: lcarva

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@tekton-robot tekton-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Nov 22, 2023
@tekton-robot
tekton-robot merged commit dd3620e into tektoncd:main Nov 22, 2023
@lcarva lcarva mentioned this pull request Dec 1, 2023
6 tasks
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants