Skip to content

🔄 Synced file(s) with valitydev/configurations - #8

Open
valitydev-bot wants to merge 3 commits into
masterfrom
update/default
Open

🔄 Synced file(s) with valitydev/configurations#8
valitydev-bot wants to merge 3 commits into
masterfrom
update/default

Conversation

@valitydev-bot

@valitydev-bot valitydev-bot commented Feb 24, 2022

Copy link
Copy Markdown

Synced local file(s) with valitydev/configurations.

Changed files
  • Synced local LICENSE with remote LICENSE
  • Created local CODEOWNERS from remote codeowners/java
  • Created local .codecov.yml from remote java/.codecov.yml

This PR was created automatically by the repo-file-sync-action workflow run #4159172700

@codecov

codecov Bot commented Feb 24, 2022

Copy link
Copy Markdown

Codecov Report

Merging #8 (33ec7ea) into master (5839a01) will not change coverage.
The diff coverage is n/a.

@@           Coverage Diff            @@
##             master      #8   +/-   ##
========================================
  Coverage      0.18%   0.18%           
  Complexity        1       1           
========================================
  Files            38      38           
  Lines           539     539           
  Branches         48      48           
========================================
  Hits              1       1           
  Misses          538     538           

Help us with your feedback. Take ten seconds to tell us how you rate us. Have a feature suggestion? Share it here.

@valitydev-bot
valitydev-bot force-pushed the update/default branch 3 times, most recently from 7553a00 to 42d9bbd Compare February 24, 2022 22:26
@valitydev-bot valitydev-bot changed the title 🔄 Synced file(s) with valitydev/.github 🔄 Synced file(s) with valitydev/configurations Feb 28, 2022
@valitydev-bot
valitydev-bot force-pushed the update/default branch 21 times, most recently from 103d10d to 8803453 Compare March 7, 2022 10:07
@valitydev-bot
valitydev-bot force-pushed the update/default branch 3 times, most recently from 515923e to 80f2730 Compare March 8, 2022 10:07
@valitydev-bot
valitydev-bot force-pushed the update/default branch 22 times, most recently from 26574e3 to 9af6b27 Compare March 20, 2022 10:06
@valitydev-bot
valitydev-bot force-pushed the update/default branch 5 times, most recently from f27bf56 to abacc22 Compare March 22, 2022 00:28
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant